Repository navigation
v3: new build system - #1265
v3: new build system#1265
Conversation
🦋 Changeset detectedLatest commit: b43ff4e The changes in this PR will be included in the next version bump. Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
d6f6abb to
328d03f
Compare
|
commit: |
019f70c to
8767553
Compare
|
I love this! Trigger to the 🌕️ |
…path as the collected external
…’t work in config files
* upgrade @opentelemetry packages to the latest versions * remove v2 only packages, will be moved to a dedicated repo * remove more v2 code and run pnpm install * use the npm yalt package in the webapp * convert @trigger.dev/core to tshy * Switch from jest to vitest in @trigger.dev/core * Fixed core test * move core-backend code into core subpath export * convert @trigger.dev/sdk to tshy * Removed hono * move core-apps to core/v3/apps, remove core-apps, start converting cli-v3 * Fix up some of the commands * cli now building and loadable * using package-json-from-dist to get package version now in core and cli * dev command WIP * cleaned up some repetition and structure of the entry point stuff * bringing back the background worker stuff * Indexing of the v3 catalog * getting closer to executing dev runs... * centralize dev logging using event emitter * Move indexing to it’s own entry point, simplify code * dev runs working * Get instrumentation to work with openai * debugging achieved internally * provide worker files as part of the worker creation on the server * support for cjs and esm javascript * Fixed timeout * worker manifest now has the config path * auto-upgrade config to non-deprecated alternatives * Adding package preview release * deployment WIP * improve the syncEnvVars output and adapt resolveEnvVars * WIP bun runtime * WIP bun support * seed tasks with the machine preset if listed in the config * deploy run executions WIP, extracted TaskRunProcess into 1 place * deployed tasks running and executing 🎉 * support for waits and better flushing & process cleanup * Fixed the heartbeating * Better warning messages * Improve and unify the indexing between dev and deploy * Support for external deps that need node-gyp to build * build extensions can now install custom packages and run instructions in the image. Also prisma extension now works and also works with multiple schema files * Add back in the main/types/module to sdk * dev no longer is Ink/React, grace period for disconnections in dev * Fix the changeset config * More changeset fixes * Remove config packages * More changeset fixes * Fixed typescript issues (needed to revert back to zod 3.22.3 * Fix pr_checks workflow * Remove the prepare script * Fixed tests and package versions * Remove cli test script * Remove packages from tailwind watch paths * Add repo to public packages * Just commit the generated files and do the building at dev time * Try and get pkg.pr.new working * Try again * Fix emitDecoratorMetadata importing named export from typescript * config file backwards compat with export const config * Fixed issue where import errors weren’t coming through * p-retry is a prod dep * typescript needs to be a prod dependency for emitDecoratorMetadata * Add better debug logging to help track down import-in-the-middle bug * An external is only considered resolvable if it resolves to the same path as the collected external * Fix runtime checks to allow >=18.20 * Move extensions to a new build package * Fixed building packages in dockerfile * Remove the e2e test from publish workflow for now * Don’t treat pkg.pr.new versions has needing upgrading * making sure config handleError works, and discovered path aliases don’t work in config files * Strip empty string env vars so they accidentally override real values * Couple of things * Update version to use preview instead of beta * Hopefully fix re-attempts with >30s delay * Match socket emit messages to current latest in main * Initial guide * Go back to beta * Go back to the preview, and update guide to use pr preview tags * Go back to beta --------- Co-authored-by: Matt Aitken <matt@mattaitken.com>
|
This breaks the config with I see the config option and the code for injecting ca certs is still there but the option has been removed in Also noticed that the file is not being copied from the config, potentially because of the I see the quickest option is to populate buildEnvVars which is not injecting anything yet. potentially missed from SyncEnvVars extension? Update I could use combination of |
This PR overhauls the build system, as well as the
devanddeploycommands.The major changes to the build system:
dependenciesToBundlebecomesexternals.additionalFilesconfig option is deprecated in favor of theadditionalFilesbuild extensionadditionalPackagesconfig option is deprecated in favor of theadditionalPackagesbuild extensionprismabuild extension to automatically include your schema.prisma file in deployed builds, as well as generating the clientresolveEnvVarshas been deprecated in favor of thesyncEnvVarsbuild extensionjsxsettings to prevent issues with having toimport * as React from "react".jsxis set to automatic by default.Other major changes included in this PR:
bunas a runtime, instead of just Node.js. The CLI still needs to execute with Node.js but tasks will run as abunsubprocess. Known issues include the lack of support for instrumentation of 3rd party SDKs.emitDecoratorMetadatabuild extension.devcommand.Some big repo changes as well:
@trigger.dev/core-appsand@trigger.dev/core-backendinto@trigger.dev/coreas subpath exports@opentelemetry-js/*packages to the latest versions.Still todo:
wasmsupport