Skip to content

Forms: use a placeholder attribute in the editor instead of value - #41712

Merged
edanzer merged 3 commits into
trunkfrom
update/forms-placeholder-attr-in-editor
Feb 27, 2025
Merged

Forms: use a placeholder attribute in the editor instead of value#41712
edanzer merged 3 commits into
trunkfrom
update/forms-placeholder-attr-in-editor

Conversation

@simison

@simison simison commented Feb 11, 2025

Copy link
Copy Markdown
Member

Use a placeholder attribute in the editor for placeholder text instead of value attribute.

By using value, we have a couple issues.

When theme includes styling for ::placeholder pseudo element, we don't show it. Consider for example CSS:

input::placeholder {
	font-style: italic;
}

Before

Screenshot 2025-02-11 at 12 49 02

After

Screenshot 2025-02-11 at 12 48 46

When we add a number input, its type="number" disallows non-numeric value, even if placeholder could be numeric. We would only need to support numeric placeholders for that input. We could change that field to type="text" in the editor, but that would miss out browser number controls, any styling applied to number fields, and otherwise also mismatch what they see in the editor with frontend.

Before

Screenshot 2025-02-11 at 12 57 28 Screenshot 2025-02-11 at 12 58 01

After

Screenshot 2025-02-11 at 12 33 56

Proposed changes:

  • Use a placeholder attribute in the editor for placeholder text instead of value attribute.
  • Use value attribute for entering the placeholder value while block is focused

Other information:

  • Have you written new tests for your changes, if applicable?
  • Have you checked the E2E test CI results, and verified that your changes do not break them?
  • Have you tested your changes on WordPress.com, if applicable (if so, you'll see a generated comment below with a script to run)?

Jetpack product discussion

Does this pull request change what data or activity we track or use?

Testing instructions:

  • Add different inputs and set their placeholders; they continue working.
  • Try different form styles; you can still enter placeholders in editor with "animated" style, but with "outlined" style you can't. That's pre-existing issue. You can use the sidebar input, though.
    image

@simison simison added Enhancement Changes to an existing feature — removing, adding, or changing parts of it [Status] Needs Review This PR is ready for review. [Package] Forms [Feature] Forms Blocks Blocks designed to streamline user input and engagement, such as contact, newsletter sign-ups, etc. labels Feb 11, 2025
@simison
simison requested a review from a team February 11, 2025 11:00
@github-actions

github-actions Bot commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

Are you an Automattician? Please test your changes on all WordPress.com environments to help mitigate accidental explosions.

  • To test on WoA, go to the Plugins menu on a WoA dev site. Click on the "Upload" button and follow the upgrade flow to be able to upload, install, and activate the Jetpack Beta plugin. Once the plugin is active, go to Jetpack > Jetpack Beta, select your plugin (Jetpack), and enable the update/forms-placeholder-attr-in-editor branch.
  • To test on Simple, run the following command on your sandbox:
bin/jetpack-downloader test jetpack update/forms-placeholder-attr-in-editor

Interested in more tips and information?

  • In your local development environment, use the jetpack rsync command to sync your changes to a WoA dev blog.
  • Read more about our development workflow here: PCYsg-eg0-p2
  • Figure out when your changes will be shipped to customers here: PCYsg-eg5-p2

@github-actions github-actions Bot added [Block] Form Form block (also see Form package label) [Feature] Forms labels Feb 11, 2025
@github-actions

github-actions Bot commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

Thank you for your PR!

When contributing to Jetpack, we have a few suggestions that can help us test and review your patch:

  • ✅ Include a description of your PR changes.
  • ✅ Add a "[Status]" label (In Progress, Needs Review, ...).
  • ✅ Add a "[Type]" label (Bug, Enhancement, Janitorial, Task).
  • ✅ Add testing instructions.
  • ✅ Specify whether this PR includes any changes to data or privacy.
  • ✅ Add changelog entries to affected projects

This comment will be updated as you work on your PR and make changes. If you think that some of those checks are not needed for your PR, please explain why you think so. Thanks for cooperation 🤖


The e2e test report can be found here. Please note that it can take a few minutes after the e2e tests checks are complete for the report to be available.


Follow this PR Review Process:

  1. Ensure all required checks appearing at the bottom of this PR are passing.
  2. Make sure to test your changes on all platforms that it applies to. You're responsible for the quality of the code you ship.
  3. You can use GitHub's Reviewers functionality to request a review.
  4. When it's reviewed and merged, you will be pinged in Slack to deploy the changes to WordPress.com simple once the build is done.

If you have questions about anything, reach out in #jetpack-developers for guidance!

@simison simison mentioned this pull request Feb 11, 2025
3 tasks
@github-actions

Copy link
Copy Markdown
Contributor

Code Coverage Summary

This PR did not change code coverage!

That could be good or bad, depending on the situation. Everything covered before, and still is? Great! Nothing was covered before? Not so great. 🤷

Full summary · PHP report · JS report

@enejb

enejb commented Feb 11, 2025

Copy link
Copy Markdown
Member

@simison Does this work as expected for you in any text input field?
I am not able to get it to work. It replaces the first character only.

lezama
lezama previously approved these changes Feb 11, 2025

@lezama lezama left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

makes sense

@lezama

lezama commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

works ok for me

It replaces the first character only.

what do you mean? 😆

@edanzer

edanzer commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

I see what @enejb is talking about. With this PR loaded, I can add placeholder text as expected in the sidebar. But if I try to type placeholder text directly into the input in the main part of the block editor, it always deletes all but the last character I typed. I also confirmed this does not happen on trunk. I'm sure it's just some mix-up of state - basically since we changed the relationship between value and placeholder, we don't track value properly when typing into the input.

@simison

simison commented Feb 11, 2025

Copy link
Copy Markdown
Member Author

Weird, that's the scenario that worked for me just fine, but I'll do further testing 👍 Might be browser dependant.

@edanzer

edanzer commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

Just to be sure we're talking and seeing the same thing, here's a video of the behavior. I type first in the placholder field in the sidebar (works), and then directly in the input.

I've also confirmed that that if I keep value = { placeholder } like below, the issue is resolved

value={ placeholder }
placeholder={ placeholder }
input-placeholder.mp4

@simison

simison commented Feb 11, 2025

Copy link
Copy Markdown
Member Author

I've also confirmed that that if I keep value = { placeholder } like below, the issue is resolved

Thanks! It might make sense to retain value when focused on the input, and set it empty when not. 🤔 That would fix the styling (keeping value removes the styles), and would still likely need some changes for inputs like Number where value can be only number.

@ntsekouras

Copy link
Copy Markdown
Member

I think we're trying to mix two different concepts in general:

  1. placeholder
  2. default value

And that was actually quite confusing for me when I first interacted with the blocks. When I started typing I thought that would be a default value..

If I'm not wrong the input fields have a prop defaultValue but it's only used in checkbox. I think we should consider allowing both default values and placeholders and have a different way of editing the placeholder (either in Inspector controls as is now, or even in block toolbar).

That would require some design input too to try to ensure that it's clear when typing in an input, it's the default value. A placeholder of course like Enter default value would be needed, but it might be deemed better to have the input uneditable and have a separate control for that too.

@simison simison mentioned this pull request Feb 14, 2025
8 tasks
@simison simison added [Status] In Progress DO NOT MERGE don't merge it! and removed [Status] Needs Review This PR is ready for review. labels Feb 14, 2025
@simison
simison force-pushed the update/forms-placeholder-attr-in-editor branch from f0212c6 to f2481f9 Compare February 24, 2025 13:22
@jp-launch-control

jp-launch-control Bot commented Feb 24, 2025

Copy link
Copy Markdown

Code Coverage Summary

This PR did not change code coverage!

That could be good or bad, depending on the situation. Everything covered before, and still is? Great! Nothing was covered before? Not so great. 🤷

Full summary · PHP report · JS report

@simison
simison force-pushed the update/forms-placeholder-attr-in-editor branch from f2481f9 to 0ca873f Compare February 27, 2025 14:56
style={ fieldStyle }
type={ type }
value={ placeholder }
type={ isSelected ? 'text' : type }

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is so that you can enter text placeholders for number input. Otherwise, you could only enter numbers.

The downside is that you might lose input[type="number"] specific styling:

Screenshot 2025-02-27 at 17 02 30

@edanzer edanzer left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Tested text and number inputs. Works great.

@edanzer
edanzer merged commit 9d31d55 into trunk Feb 27, 2025
@edanzer
edanzer deleted the update/forms-placeholder-attr-in-editor branch February 27, 2025 19:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

[Block] Form Form block (also see Form package label) DO NOT MERGE don't merge it! Enhancement Changes to an existing feature — removing, adding, or changing parts of it [Feature] Forms Blocks Blocks designed to streamline user input and engagement, such as contact, newsletter sign-ups, etc. [Feature] Forms [Package] Forms

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants