Skip to content

fix: prevent prerender crawler from hanging on unterminated unquoted attributes - #17141

Merged
Rich-Harris merged 1 commit into
version-3from
fix-crawler
Sep 18, 2026
Merged

Rich-Harris merged 1 commit into
version-3from
fix-crawler

Conversation

@Rich-Harris

Copy link
Copy Markdown
Member

Fixes a hypothetical bug with the crawler — if a document ends with an unquoted attribute, the crawler will hang. No test because it's essentially impossible to end up in this situation, but worth fixing to avoid tripping up automated bug scanners (which is how this surfaced)


Please don't delete this checklist! Before submitting the PR, please make sure you do the following:

  • It's really useful if your PR references an issue where it is discussed ahead of time. In many cases, features are absent for a reason. For large changes, please create an RFC: https://github.com/sveltejs/rfcs
  • This message body should clearly illustrate what problems it solves.
  • Ideally, include a test that fails without this PR but passes with it.

Tests

  • Run the tests with pnpm test and lint the project with pnpm lint and pnpm check

Changesets

  • If your PR makes a change that should be noted in one or more packages' changelogs, generate a changeset by running pnpm changeset and following the prompts. Changesets that add features should be minor and those that fix bugs should be patch. Please prefix changeset messages with feat:, fix:, or chore:.

Edits

  • Please ensure that 'Allow edits from maintainers' is checked. PRs without this option may be closed.

@pkg-svelte-dev

Copy link
Copy Markdown

Install the latest version of @sveltejs/kit from 0e1f09c:

pnpm add https://pkg.svelte.dev/@sveltejs/kit/c/0e1f09cdbd5aacaa0bbec60ccdecfbc05389eee0

Open in pkg.svelte.dev: https://pkg.svelte.dev/repos/kit/pr/17141

@changeset-bot

changeset-bot Bot commented Sep 18, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0e1f09c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@sveltejs/kit Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@svelte-docs-bot

Copy link
Copy Markdown

@Rich-Harris
Rich-Harris merged commit b21766f into version-3 Sep 18, 2026
41 checks passed
@Rich-Harris
Rich-Harris deleted the fix-crawler branch September 18, 2026 18:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants