Skip to content

build(deps): bump vercel from 62.7.0 to 63.1.0 in /docs-website - #13195

Open
dependabot[bot] wants to merge 2 commits into
mainfrom
dependabot/npm_and_yarn/docs-website/vercel-63.1.0
Open

dependabot[bot] wants to merge 2 commits into
mainfrom
dependabot/npm_and_yarn/docs-website/vercel-63.1.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 9, 2026

Copy link
Copy Markdown
Contributor

Bumps vercel from 62.7.0 to 63.1.0.

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [vercel](https://github.com/vercel/vercel/tree/HEAD/packages/cli) from 62.7.0 to 63.1.0.
- [Release notes](https://github.com/vercel/vercel/releases)
- [Changelog](https://github.com/vercel/vercel/blob/main/packages/cli/CHANGELOG.md)
- [Commits](https://github.com/vercel/vercel/commits/HEAD/packages/cli)

---
updated-dependencies:
- dependency-name: vercel
  dependency-version: 63.1.0
  dependency-type: direct:production
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Oct 9, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner October 9, 2026 14:25
@dependabot
dependabot Bot requested review from anakin87 and removed request for a team October 9, 2026 14:25
@vercel

vercel Bot commented Oct 9, 2026 •

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
haystack-docs Ignored Ignored Preview Oct 9, 2026 3:13pm UTC

Request Review

This branch had an error being deployed

1 failed (outdated) deployment
Preview — 49e4122d Deployed Oct 9, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant