Skip to content

fix: (queryInstance.clearInterval) is not called on unmount - #717

Merged
tannerlinsley merged 1 commit into
TanStack:masterfrom
otaciliolacerda:Issue-711
Jul 8, 2020
Merged

tannerlinsley merged 1 commit into
TanStack:masterfrom
otaciliolacerda:Issue-711

Conversation

@otaciliolacerda

Copy link
Copy Markdown
Contributor

Fix #711

@vercel

vercel Bot commented Jul 8, 2020 •

Copy link
Copy Markdown

This pull request is being automatically deployed with Vercel (learn more).
To see the status of your deployment, click below or on the icon next to each commit.

🔍 Inspect: https://vercel.com/tannerlinsley/react-query/mhoufij0r
✅ Preview: https://react-query-git-fork-otaciliolacerda-issue-711.tannerlinsley.vercel.app

@tannerlinsley

Copy link
Copy Markdown
Member

🎉 This PR is included in version 2.4.14 🎉

The release is available on:

Your semantic-release bot 📦🚀

@johnsetzer

Copy link
Copy Markdown

Impressive turn around time. Nice work.

@nx-cloud

nx-cloud Bot commented Jun 24, 2026 •

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit d02bae9

Command Status Duration Result
nx run-many --target=build --exclude=examples/*... ✅ Succeeded 2m 3s View ↗
nx affected --targets=test:sherif,test:knip,tes... ✅ Succeeded <1s View ↗

☁️ Nx Cloud last updated this comment at 2026-06-24 12:35:28 UTC

This branch was successfully deployed

1 active deployment
Preview — d02bae91 Deployed Jul 8, 2020 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

queryInstance.clearInterval() is not called on Unmount

3 participants