Skip to content

Auto-activate ruby manifest tests - #6294

Closed
dd-octo-sts[bot] wants to merge 1 commit into
mainfrom
auto-ruby-manifest
Closed

dd-octo-sts[bot] wants to merge 1 commit into
mainfrom
auto-ruby-manifest

Conversation

@dd-octo-sts

@dd-octo-sts dd-octo-sts Bot commented Feb 13, 2026

Copy link
Copy Markdown
Contributor

Automated activation of easy-win tests for ruby
View nightly workflow run

  • If you approve this PR please also merge it.
  • If the tests are failing it might be due to a change made since the last nightly system-tests run. You can close the PR, an updated one will be available tomorrow.
  • If you close the PR please also delete the branch

This commit was created from the local commit with hash 98f80e723f9c801a7b5f3a74868b882f84b99b6b.
@dd-octo-sts
dd-octo-sts Bot requested review from a team as code owners February 13, 2026 06:59
@dd-octo-sts
dd-octo-sts Bot enabled auto-merge (squash) February 13, 2026 06:59
@github-actions

Copy link
Copy Markdown
Contributor

CODEOWNERS have been resolved as:

manifests/ruby.yml                                                      @DataDog/ruby-guild @DataDog/asm-ruby

@Strech Strech closed this Feb 13, 2026
auto-merge was automatically disabled February 13, 2026 09:18

Pull request was closed

@cbeauchesne
cbeauchesne deleted the auto-ruby-manifest branch February 21, 2026 11:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant