Skip to content

DI: enable ruby process tags test - #6292

Merged
p-datadog merged 1 commit into
mainfrom
enable-ruby
Feb 13, 2026
Merged

p-datadog merged 1 commit into
mainfrom
enable-ruby

Conversation

@p-datadog

@p-datadog p-datadog commented Feb 12, 2026 •

Copy link
Copy Markdown
Member

Motivation

Checking which existing tests can be enabled for Ruby

Changes

Removes ruby exclusion from the process tags test

Workflow

  1. ⚠️ Create your PR as draft ⚠️
  2. Work on you PR until the CI passes
  3. Mark it as ready for review
    • Test logic is modified? -> Get a review from RFC owner.
    • Framework is modified, or non obvious usage of it -> get a review from R&P team

🚀 Once your PR is reviewed and the CI green, you can merge it!

🛟 #apm-shared-testing 🛟

Reviewer checklist

  • Anything but tests/ or manifests/ is modified ? I have the approval from R&P team
  • A docker base image is modified?
    • the relevant build-XXX-image label is present
  • A scenario is added, removed or renamed?

@p-datadog
p-datadog requested review from a team as code owners February 12, 2026 17:20
@github-actions

Copy link
Copy Markdown
Contributor

CODEOWNERS have been resolved as:

tests/debugger/test_debugger_probe_snapshot.py                          @DataDog/debugger @DataDog/system-tests-core

@p-datadog
p-datadog merged commit de5b3bf into main Feb 13, 2026
415 checks passed
@p-datadog
p-datadog deleted the enable-ruby branch February 13, 2026 19:02
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants