Repository navigation
feat: withdraw - #78
Open
Gustavonobregab wants to merge 6 commits into
Open
Gustavonobregab wants to merge 6 commits into
Gustavonobregab wants to merge 6 commits into
Conversation
…atePay#48) * fix: update Pix check and simulatePayment to use id query param (AbacatePay#41) * fix: fixando testes e adicionando metadata no body
* Update CONTRIBUTING.md * Update CONTRIBUTING.md * Update README.md * chore: melhorando README.md * chore: melhorando README.md * fix: fixing README.md image url * Update README.md * chore: bump version bump version * chore: fix format fix overall code format * fix: fixing pr, addig changeset file * fix: changing bot permissions --------- Co-authored-by: Christopher Ribeiro <christo_campiglia@hotmail.com> Co-authored-by: costa38r <costa38r@proton.me>
Co-authored-by: costa38r <costa38r@proton.me>
🦋 Changeset detectedLatest commit: 53cee15 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
lucasrodrigges
approved these changes
Nov 30, 2025
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Pull Request - AbacatePay SDK
Issue Relacionada
Closes #77
(Substitua pelo número da Issue, ex:
Closes #42)Descrição
Este Pull Request adiciona suporte à funcionalidade withdraw no SDK da AbacatePay.
Implementação da lógica de saque (withdraw) no src/index.ts.
Tipagem atualizada em src/types.ts.
Criação de testes unitários em test/index.test.ts para validar o comportamento do withdraw.
Garanta que:
tipo: descrição breve(ex:feat: adiciona suporte a webhook).Notas Adicionais